Skip to content

🤖 Update module github.com/stackitcloud/stackit-sdk-go/services/alb to v0.17.2 - #36

Merged
ske-prow[bot] merged 1 commit into
mainfrom
renovate/github.com-stackitcloud-stackit-sdk-go-services-alb-0.17.x
Sep 28, 2026
Merged

ske-prow[bot] merged 1 commit into
mainfrom
renovate/github.com-stackitcloud-stackit-sdk-go-services-alb-0.17.x

Conversation

@ske-renovate-operator

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
github.com/stackitcloud/stackit-sdk-go/services/alb (changelog) v0.17.1 → v0.17.2 age confidence

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@ske-renovate-operator ske-renovate-operator Bot added the kind/upgrade Upgrading components or dependencies (e.g., by renovate) label Sep 8, 2026
@ske-renovate-operator
ske-renovate-operator Bot requested a review from a team September 8, 2026 08:31
@ske-prow ske-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 8, 2026
@stackit-ske

Copy link
Copy Markdown

Our project currently lacks enough active contributors to adequately respond to all PRs.
This bot triages PRs according to the following rules:

  • After 15d of inactivity, lifecycle/stale is applied
  • After 15d of inactivity since lifecycle/stale was applied, lifecycle/rotten is applied
  • After 7d of inactivity since lifecycle/rotten was applied, the PR is closed

You can:

  • Mark this PR as fresh with /remove-lifecycle stale
  • Mark this PR as rotten with /lifecycle rotten
  • Close this PR with /close

/lifecycle stale

@ske-prow ske-prow Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Sep 23, 2026
@ske-renovate-operator
ske-renovate-operator Bot force-pushed the renovate/github.com-stackitcloud-stackit-sdk-go-services-alb-0.17.x branch from cd2f0ef to 5d42437 Compare September 28, 2026 08:43
@ske-prow ske-prow Bot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. and removed size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Sep 28, 2026
@ske-prow ske-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 28, 2026
@ske-prow

ske-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dergeberl

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ske-prow

ske-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 5e051efd5a83e5b77dc1acb1aa4a30184d5ed9fd

@ske-prow ske-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 28, 2026
@ske-prow
ske-prow Bot merged commit 5610360 into main Sep 28, 2026
3 checks passed
@ske-prow
ske-prow Bot deleted the renovate/github.com-stackitcloud-stackit-sdk-go-services-alb-0.17.x branch September 28, 2026 08:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/upgrade Upgrading components or dependencies (e.g., by renovate) lgtm Indicates that a PR is ready to be merged. lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants